build(deps): bump github.com/creack/pty from 1.1.18 to 1.1.24 - #22277
build(deps): bump github.com/creack/pty from 1.1.18 to 1.1.24#22277dependabot[bot] wants to merge 1 commit into
Conversation
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a etcd-io member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Bumps [github.com/creack/pty](https://github.com/creack/pty) from 1.1.18 to 1.1.24. - [Release notes](https://github.com/creack/pty/releases) - [Commits](creack/pty@v1.1.18...v1.1.24) --- updated-dependencies: - dependency-name: github.com/creack/pty dependency-version: 1.1.24 dependency-type: direct:production update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
b231452 to
74ec28e
Compare
See upstream PR etcd-io#22277. Signed-off-by: vivekpatani <9080894+vivekpatani@users.noreply.github.com>
|
Looks like github.com/creack/pty is up-to-date now, so this is no longer needed. |
See upstream PR etcd-io#22277. Signed-off-by: vivekpatani <9080894+vivekpatani@users.noreply.github.com>
Bumps github.com/creack/pty from 1.1.18 to 1.1.24.
Release notes
Sourced from github.com/creack/pty's releases.
... (truncated)
Commits
edfbf75Merge pull request #201 from MacMalainey/master7c00df3remove x/sys dependency for zos13c571dadd z/OS support2cde18bFix go.mod08e77a0Merge pull request #189 from n2vi/openbsd-arch-free3c9ed8dztypes_openbsd_32bit_int.go: remove arch list03db72cMerge pull request #180 from WeidiDeng/non-blockingf3f519eadd notes to README.md2711aacusing a larger timeout in testbb5f79cre-enable non-blocking tests