Skip to content

Bump go version to 1.26.0 - #21294

Merged
fuweid merged 4 commits into
etcd-io:mainfrom
ahrtr:20260211_go_1.26
Feb 11, 2026
Merged

Bump go version to 1.26.0#21294
fuweid merged 4 commits into
etcd-io:mainfrom
ahrtr:20260211_go_1.26

Conversation

@ahrtr

@ahrtr ahrtr commented Feb 11, 2026

Copy link
Copy Markdown
Member

No description provided.

Signed-off-by: Benjamin Wang <benjamin.ahrtr@gmail.com>
Signed-off-by: Benjamin Wang <benjamin.ahrtr@gmail.com>
@codecov

codecov Bot commented Feb 11, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 68.37%. Comparing base (8f1c499) to head (fd5baa4).
⚠️ Report is 7 commits behind head on main.

Additional details and impacted files

see 22 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #21294      +/-   ##
==========================================
- Coverage   68.41%   68.37%   -0.04%     
==========================================
  Files         428      428              
  Lines       35256    35256              
==========================================
- Hits        24119    24105      -14     
- Misses       9740     9760      +20     
+ Partials     1397     1391       -6     

Continue to review full report in Codecov by Sentry.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 8f1c499...fd5baa4. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

Signed-off-by: Benjamin Wang <benjamin.ahrtr@gmail.com>
Fix the following workflow failure:
FAIL: inconsistent versions for dependency: github.com/sirupsen/logrus
  - github.com/sirupsen/logrus@v1.9.3 (indirect) from: go.etcd.io/etcd/etcdutl/v3
  - github.com/sirupsen/logrus@v1.9.3 (indirect) from: go.etcd.io/etcd/server/v3
  - github.com/sirupsen/logrus@v1.9.3 (indirect) from: go.etcd.io/etcd/tests/v3
  - github.com/sirupsen/logrus@v1.9.3 (indirect) from: go.etcd.io/etcd/v3
  - github.com/sirupsen/logrus@v1.9.4 (indirect) from: go.etcd.io/etcd/tools/v3

Signed-off-by: Benjamin Wang <benjamin.ahrtr@gmail.com>
@ahrtr

ahrtr commented Feb 11, 2026

Copy link
Copy Markdown
Member Author

cc @fuweid @ivanvc

cc @liggitt once this PR gets merged, you can remove all the ptr functions in your PR, and use new(a) directly when needed.

@k8s-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ahrtr, fuweid

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@fuweid
fuweid merged commit 6b93faf into etcd-io:main Feb 11, 2026
32 checks passed
@ahrtr
ahrtr deleted the 20260211_go_1.26 branch February 11, 2026 18:54
@ivanvc

ivanvc commented Feb 11, 2026

Copy link
Copy Markdown
Member

Wow, this minor version bump required minimal changes. Glad to see. I haven't opened the issue to upgrade to 1.26, because we still have the latest 1.25 and 1.24 versions on flight, and 1.26 was released yesterday. Thanks, @ahrtr.

@ivanvc ivanvc mentioned this pull request Feb 25, 2026
18 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

4 participants