Skip to content

Add CODEOWNERS - #117

Merged
200km merged 1 commit into
developfrom
feature/add-codeowners
Jun 22, 2026
Merged

Add CODEOWNERS#117
200km merged 1 commit into
developfrom
feature/add-codeowners

Conversation

@200km

@200km 200km commented Jun 22, 2026

Copy link
Copy Markdown
Member

Add CODEOWNERS

@200km 200km self-assigned this Jun 22, 2026
@200km 200km added the priority::high priorityg high label Jun 22, 2026
@github-project-automation github-project-automation Bot moved this to 🏗 In Progress in S2E Jun 22, 2026
@200km

200km commented Jun 22, 2026

Copy link
Copy Markdown
Member Author

mdbookは今使っていないので、マージしてしまおうと思います。

@200km
200km merged commit 1dc1f9e into develop Jun 22, 2026
1 check failed
@200km
200km deleted the feature/add-codeowners branch June 22, 2026 07:14
@github-project-automation github-project-automation Bot moved this from 🏗 In Progress to 💮 Done in S2E Jun 22, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

priority::high priorityg high

Projects

Status: 💮 Done

Development

Successfully merging this pull request may close these issues.

2 participants