Skip to content

Update to Nx 14 - #77

Draft
trumbitta wants to merge 4 commits into
mainfrom
update-nx-14
Draft

Update to Nx 14#77
trumbitta wants to merge 4 commits into
mainfrom
update-nx-14

Conversation

@trumbitta

@trumbitta trumbitta commented Aug 26, 2022

Copy link
Copy Markdown
Owner

But nx-plugin-unused-deps doesn't seem to work anymore (tests don't pass) 😭

I also had to update to a version that's not the latest because I had an error with a migration.

@nx-cloud

nx-cloud Bot commented Aug 26, 2022

Copy link
Copy Markdown

☁️ Nx Cloud Report

CI is running/has finished running commands for commit 30d513f. As they complete they will appear below. Click to see the status, the terminal output, and the build insights.

📂 See all runs for this branch


🟥 Failed Commands
nx affected:e2e --base=remotes/origin/main
✅ Successfully ran 1 target

Sent with 💌 from NxCloud.

@gitpod-io

gitpod-io Bot commented Aug 26, 2022

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant