Skip to content

bump(deps): bump tektoncd/pipeline to version v1.6.3 - #2053

Open
pratap0007 wants to merge 1 commit into
tektoncd:release-v0.34.xfrom
pratap0007:bump-pipelines-0.34.x
Open

bump(deps): bump tektoncd/pipeline to version v1.6.3#2053
pratap0007 wants to merge 1 commit into
tektoncd:release-v0.34.xfrom
pratap0007:bump-pipelines-0.34.x

Conversation

@pratap0007

Copy link
Copy Markdown
Contributor

Changes

Submitter Checklist

As the author of this PR, please check off the items in this checklist:

  • Has Docs if any changes are user facing, including updates to minimum requirements e.g. Kubernetes version bumps
  • Has Tests included if any functionality added or changed
  • Follows the commit message standard
  • Meets the Tekton contributor standards (including functionality, content, code)
  • Has a kind label. You can add one by adding a comment on this PR that contains /kind <type>. Valid types are bug, cleanup, design, documentation, feature, flake, misc, question, tep
  • Release notes block below has been updated with any user facing changes (API changes, bug fixes, changes requiring upgrade notices or deprecation warnings). See some examples of good release notes.
  • Release notes contains the string "action required" if the change requires additional action from users switching to the new release

Release Notes

NONE

Signed-off-by: Shiv Verma <shverma@redhat.com>
@tekton-robot tekton-robot added the release-note-none Denotes a PR that doesnt merit a release note. label Jun 23, 2026
@tekton-robot
tekton-robot requested review from iancoffey and khrm June 23, 2026 16:26
@tekton-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To complete the pull request process, please assign iancoffey after the PR has been reviewed.
You can assign the PR to them by writing /assign @iancoffey in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Jun 23, 2026

@khrm khrm left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@pratap0007 Can you update golang-ci-lint workflow for this branch in another commit. It will fix this.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

release-note-none Denotes a PR that doesnt merit a release note. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants