Skip to content

build(deps): bump google.golang.org/api from 0.293.0 to 0.297.0 in /pipelinerun-logs in the all group across 1 directory - #3640

Merged
tekton-robot merged 1 commit into
mainfrom
dependabot/go_modules/pipelinerun-logs/all-841ad48716
Sep 8, 2026
Merged

build(deps): bump google.golang.org/api from 0.293.0 to 0.297.0 in /pipelinerun-logs in the all group across 1 directory#3640
tekton-robot merged 1 commit into
mainfrom
dependabot/go_modules/pipelinerun-logs/all-841ad48716

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 3, 2026

Copy link
Copy Markdown
Contributor

Bumps the all group with 1 update in the /pipelinerun-logs directory: google.golang.org/api.

Updates google.golang.org/api from 0.293.0 to 0.297.0

Release notes

Sourced from google.golang.org/api's releases.

v0.297.0

0.297.0 (2026-09-01)

Features

  • Move to go1.26.0 as the lowest supported go version (#3724) (7770e01)

v0.296.0

0.296.0 (2026-08-31)

Features

v0.295.0

0.295.0 (2026-08-28)

Features

v0.294.0

0.294.0 (2026-08-26)

Features

Changelog

Sourced from google.golang.org/api's changelog.

0.297.0 (2026-09-01)

Features

  • Move to go1.26.0 as the lowest supported go version (#3724) (7770e01)

0.296.0 (2026-08-31)

Features

0.295.0 (2026-08-28)

Features

0.294.0 (2026-08-26)

Features

Commits

@dependabot dependabot Bot added dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. labels Sep 3, 2026
@tekton-robot tekton-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 3, 2026
@tekton-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: AlanGreene

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 4, 2026
@anithapriyanatarajan

Copy link
Copy Markdown
Contributor

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 8, 2026
Bumps the all group with 1 update in the /pipelinerun-logs directory: [google.golang.org/api](https://github.com/googleapis/google-api-go-client).


Updates `google.golang.org/api` from 0.293.0 to 0.297.0
- [Release notes](https://github.com/googleapis/google-api-go-client/releases)
- [Changelog](https://github.com/googleapis/google-api-go-client/blob/main/CHANGES.md)
- [Commits](googleapis/google-api-go-client@v0.293.0...v0.297.0)

---
updated-dependencies:
- dependency-name: google.golang.org/api
  dependency-version: 0.296.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: all
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot changed the title build(deps): bump google.golang.org/api from 0.293.0 to 0.296.0 in /pipelinerun-logs in the all group build(deps): bump google.golang.org/api from 0.293.0 to 0.297.0 in /pipelinerun-logs in the all group across 1 directory Sep 8, 2026
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/pipelinerun-logs/all-841ad48716 branch from 9de6f95 to 08f5de5 Compare September 8, 2026 02:28
@tekton-robot tekton-robot removed the lgtm Indicates that a PR is ready to be merged. label Sep 8, 2026
@anithapriyanatarajan

Copy link
Copy Markdown
Contributor

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 8, 2026
@tekton-robot
tekton-robot merged commit 8a09b6d into main Sep 8, 2026
8 checks passed
@dependabot
dependabot Bot deleted the dependabot/go_modules/pipelinerun-logs/all-841ad48716 branch September 8, 2026 03:38
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dependencies Used by dependabot - identifies all PRs created by dependabot kind/misc Categorizes issue or PR as a miscellaneuous one. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. release-note-none Denotes a PR that doesnt merit a release note. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants