Skip to content

chore: delete dead code in test/per_feature_flags_test.go file - #10802

Merged
tekton-robot merged 1 commit into
tektoncd:mainfrom
zakisk:delete-dead-code
Sep 30, 2026
Merged

tekton-robot merged 1 commit into
tektoncd:mainfrom
zakisk:delete-dead-code

Conversation

@zakisk

@zakisk zakisk commented Sep 24, 2026

Copy link
Copy Markdown
Member

Changes

test/per_feature_flags_test.go file has system namespace retrieve code from env and fallback to hardcoded value but that ns value is never used in code so deletes it.

Submitter Checklist

As the author of this PR, please check off the items in this checklist:

  • Has Docs if any changes are user facing, including updates to minimum requirements e.g. Kubernetes version bumps
  • Has Tests included if any functionality added or changed
  • pre-commit Passed
  • Follows the commit message standard
  • Meets the Tekton contributor standards (including functionality, content, code)
  • Has a kind label. You can add one by adding a comment on this PR that contains /kind <type>. Valid types are bug, cleanup, design, documentation, feature, flake, misc, question, tep
  • Release notes block below has been updated with any user facing changes (API changes, bug fixes, changes requiring upgrade notices or deprecation warnings). See some examples of good release notes.
  • Release notes contains the string "action required" if the change requires additional action from users switching to the new release

Release Notes

NOT REQUIRED

@tekton-robot tekton-robot added the release-note Denotes a PR that will be considered when it comes time to generate release notes. label Sep 24, 2026
@tekton-robot tekton-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 24, 2026
@zakisk

zakisk commented Sep 24, 2026

Copy link
Copy Markdown
Member Author

/kind chore

@tekton-robot

Copy link
Copy Markdown
Collaborator

@zakisk: The label(s) kind/chore cannot be applied, because the repository doesn't have them.

Details

In response to this:

/kind chore

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@zakisk

zakisk commented Sep 24, 2026

Copy link
Copy Markdown
Member Author

/kind cleanup

@tekton-robot tekton-robot added the kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. label Sep 24, 2026
@codecov

codecov Bot commented Sep 24, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 90.70%. Comparing base (538955f) to head (e03a67d).

Additional details and impacted files
@@           Coverage Diff           @@
##             main   #10802   +/-   ##
=======================================
  Coverage   90.70%   90.70%           
=======================================
  Files         295      295           
  Lines       22535    22535           
=======================================
  Hits        20440    20440           
  Misses       2093     2093           
  Partials        2        2           
Flag Coverage Δ
unit-tests 90.70% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@zakisk
zakisk force-pushed the delete-dead-code branch 3 times, most recently from f27fbd2 to 500f734 Compare September 28, 2026 10:48
test/per_feature_flags_test.go file has system namespace retrieve
code from env and fallback to hardcoded value but that ns value
is never used in code so deletes it.

Signed-off-by: Zaki Shaikh <zashaikh@redhat.com>

@vdemeester vdemeester left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@tekton-robot tekton-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 30, 2026
@tekton-robot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vdemeester

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@tekton-robot tekton-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 30, 2026
@tekton-robot
tekton-robot merged commit 61e0221 into tektoncd:main Sep 30, 2026
44 of 47 checks passed
@zakisk
zakisk deleted the delete-dead-code branch September 30, 2026 10:50
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm Indicates that a PR is ready to be merged. release-note Denotes a PR that will be considered when it comes time to generate release notes. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants