Skip to content

chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (release-1.9) - #2318

Merged
rm3l merged 1 commit into
release-1.9from
renovate/release-1.9-dockerfile-all-dockerfile-images
Feb 11, 2026
Merged

chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (release-1.9)#2318
rm3l merged 1 commit into
release-1.9from
renovate/release-1.9-dockerfile-all-dockerfile-images

Conversation

@renovate

@renovate renovate Bot commented Feb 11, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi9/go-toolset stage patch 9.7-17705965859.7-1770654497

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

… tag to v9.7-1770654497

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@sonarqubecloud

Copy link
Copy Markdown

@polasudo

Copy link
Copy Markdown
Collaborator

/lgtm
/approve

@rm3l rm3l left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Feb 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: polasudo, rm3l

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@rm3l
rm3l merged commit ef9f461 into release-1.9 Feb 11, 2026
9 of 10 checks passed
@rm3l
rm3l deleted the renovate/release-1.9-dockerfile-all-dockerfile-images branch February 11, 2026 08:44
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants