Skip to content

chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (main) - autoclosed - #2312

Closed
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/main-dockerfile-all-dockerfile-images
Closed

chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (main) - autoclosed#2312
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/main-dockerfile-all-dockerfile-images

Conversation

@renovate

@renovate renovate Bot commented Feb 11, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
registry.access.redhat.com/ubi9/go-toolset stage patch 9.7-17705965859.7-1770654497

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

… tag to v9.7-1770654497

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@openshift-ci

openshift-ci Bot commented Feb 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign rm3l for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sonarqubecloud

Copy link
Copy Markdown

@renovate renovate Bot changed the title chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (main) chore(deps): update registry.access.redhat.com/ubi9/go-toolset docker tag to v9.7-1770654497 (main) - autoclosed Feb 11, 2026
@renovate renovate Bot closed this Feb 11, 2026
@renovate
renovate Bot deleted the renovate/main-dockerfile-all-dockerfile-images branch February 11, 2026 09:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/dependency upgrade Dependency upgrades

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants