Skip to content

fix(deps): update k8s.io/utils digest to b8788ab (release-1.7) - #2309

Closed
renovate[bot] wants to merge 1 commit into
release-1.7from
renovate/release-1.7-k8s.io-utils-digest
Closed

fix(deps): update k8s.io/utils digest to b8788ab (release-1.7)#2309
renovate[bot] wants to merge 1 commit into
release-1.7from
renovate/release-1.7-k8s.io-utils-digest

Conversation

@renovate

@renovate renovate Bot commented Feb 10, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
k8s.io/utils require digest 4c0f3b2b8788ab

Warning

Some dependencies could not be looked up. Check the Dependency Dashboard for more information.


Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
@renovate renovate Bot added the kind/dependency upgrade Dependency upgrades label Feb 10, 2026
@openshift-ci

openshift-ci Bot commented Feb 10, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign gazarenkov for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sonarqubecloud

Copy link
Copy Markdown

@rm3l rm3l closed this Feb 10, 2026
@renovate

renovate Bot commented Feb 10, 2026

Copy link
Copy Markdown
Contributor Author

Renovate Ignore Notification

Because you closed this PR without merging, Renovate will ignore this update. You will not get PRs for the k8s.io/utils b8788ab update again.

If you accidentally closed this PR, or if you changed your mind: rename this PR to get a fresh replacement PR.

@renovate
renovate Bot deleted the renovate/release-1.7-k8s.io-utils-digest branch February 10, 2026 21:53
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/dependency upgrade Dependency upgrades

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant