Skip to content

Generate schemas - #136

Open
github-actions[bot] wants to merge 1 commit into
mainfrom
generate-schemas
Open

Generate schemas#136
github-actions[bot] wants to merge 1 commit into
mainfrom
generate-schemas

Conversation

@github-actions

@github-actions github-actions Bot commented Jun 7, 2023

Copy link
Copy Markdown
Contributor

This PR is automatically generated. It updates schemas in the repository.
More information on how the schemas are updated can be found in the README.

@github-actions github-actions Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. ok-to-test labels Jun 7, 2023
@op1st-prow op1st-prow Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Jun 7, 2023
@op1st-prow

op1st-prow Bot commented Jun 7, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by:

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@op1st-prow

op1st-prow Bot commented Jun 7, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by:

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@op1st-prow

op1st-prow Bot commented Jun 7, 2023

Copy link
Copy Markdown

Hi @github-actions[bot]. Thanks for your PR.

I'm waiting for a operate-first member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. ok-to-test size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant