Repository navigation
chore(deps): update module github.com/emicklei/go-restful to v3 - #1086
red-hat-konflux-kflux-prd-rh03[bot] wants to merge 1 commit into
Conversation
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configuration
You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: red-hat-konflux-kflux-prd-rh03[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @red-hat-konflux-kflux-prd-rh03[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## master #1086 +/- ##
=======================================
Coverage 48.39% 48.39%
=======================================
Files 46 46
Lines 7271 7271
=======================================
Hits 3519 3519
Misses 3390 3390
Partials 362 362 🚀 New features to boost your workflow:
|
51c905a to
619c2e6
Compare
Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com> Signed-off-by: red-hat-konflux-kflux-prd-rh03 <206760901+red-hat-konflux-kflux-prd-rh03[bot]@users.noreply.github.com>
619c2e6 to
6270ca3
Compare
|
@red-hat-konflux-kflux-prd-rh03[bot]: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
This downgraded, that makes no sense, closing. |
Renovate Ignore NotificationBecause you closed this PR without merging, Renovate will ignore this update. You will not get PRs for any future If you accidentally closed this PR, or if you changed your mind: rename this PR to get a fresh replacement PR. |
This PR contains the following updates:
v2.16.0+incompatible→v3.13.0Release Notes
emicklei/go-restful (github.com/emicklei/go-restful)
v3.13.0Compare Source
v3.12.2Compare Source
v3.12.1Compare Source
v3.12.0Compare Source
v3.11.3Compare Source
v3.11.2Compare Source
v3.11.1Compare Source
v3.11.0Compare Source
v3.10.2Compare Source
see comment in Readme how to customize this behaviour.
v3.10.1Compare Source
v3.10.0Compare Source
v3.9.0Compare Source
v3.8.0Compare Source
by changing the behaviour of the AllowedDomains setting in the CORS filter.
To support the previous behaviour, the CORS filter type now has a AllowedDomainFunc
callback mechanism which is called when a simple domain match fails.
v3.7.4Compare Source
v3.7.3Compare Source
v3.7.2Compare Source
v3.7.1Compare Source
v3.7.0Compare Source
v3.6.0Compare Source
v3.5.2Compare Source
v3.5.1Compare Source
v3.5.0Compare Source
v3.4.0Compare Source
v3.3.3Compare Source
v3.3.1Compare Source
v3.3.0Compare Source
v3.2.0Compare Source
v3.1.0Compare Source
v3.0.0
Module consumer should use github.com/emicklei/go-restful/v3 as import path
v2.10.0
v2.9.6
v2.11.1
v2.11.0
v2.9.6
v2.9.5
v2.9.4
v2.9.3
v2.9.2
v2.9.1
v2.9.0
v2.8.0
v2.7.1
v2.6.1
v2.6.0
2017-09-13
.If(func)in route building.2017-02-16
2017-01-30
2017-01-27
2016-12-22
2016-11-26
2016-09-22
2016-02-14
2015-09-27
2015-09-25
2015-09-14
2015-08-06
2015-03-20
2015-03-18
2015-03-17
2014-11-27
2014-11-12
2014-11-10
2014-10-31
2014-10-23
2014-04-16
2014-06-27
2014-07-03
2014-03-12
2014-02-26
2014-02-17
2014-01-10
2014-01-07
2013-11-13
2013-10-29
2013-10-04
2013-09-12
2013-08-05
2013-08-27
2014-04-15
2013-08-08
Important API changes:
2013-07-06
2013-06-19
2013-06-03
2013-06-02
2013-05-22
2013-05-18
[2012-11-14 .. 2013-05-18>
2012-11-14
v3.0.1Compare Source
v3.0.0Compare Source
Configuration
📅 Schedule: (in timezone UTC)
* 2-4 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
To execute skipped test pipelines write comment
/ok-to-test.Documentation
Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.