Skip to content

[release-v1.23] Update Konflux configurations - #885

Open
serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23
Open

[release-v1.23] Update Konflux configurations#885
serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Update Konflux configurations

@openshift-ci
openshift-ci Bot requested review from aliok and creydr August 24, 2026 10:51
@openshift-ci

openshift-ci Bot commented Aug 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign lberk for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 3 times, most recently from cb36983 to 27371db Compare August 31, 2026 13:16
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 15 times, most recently from ece5420 to 067604b Compare September 7, 2026 11:58
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 2 times, most recently from 5d3244e to b5d45c0 Compare September 8, 2026 08:51
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch from b5d45c0 to e343b4a Compare September 8, 2026 18:17
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant