[release-v1.22] Update Konflux configurations - #882
Open
serverless-qe wants to merge 1 commit into
Open
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
from
July 20, 2026 09:28
87c8156 to
a1b52f9
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
2 times, most recently
from
August 3, 2026 10:15
e0d4e96 to
52ae78d
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
2 times, most recently
from
August 17, 2026 07:24
9efcdad to
26e4159
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
5 times, most recently
from
August 26, 2026 14:02
2ff35c1 to
4b46dff
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
16 times, most recently
from
September 7, 2026 11:58
e52e84e to
cb0ff7e
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
from
September 7, 2026 18:12
cb0ff7e to
63485b6
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
from
September 8, 2026 08:51
63485b6 to
683ccfc
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.22
branch
from
September 8, 2026 18:17
683ccfc to
ad50725
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Update Konflux configurations