Skip to content

chore(deps-dev): bump @happy-dom/global-registrator from 20.11.2 to 20.12.0 in /tests - #2697

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/tests/happy-dom/global-registrator-20.12.0
Open

chore(deps-dev): bump @happy-dom/global-registrator from 20.11.2 to 20.12.0 in /tests#2697
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/npm_and_yarn/tests/happy-dom/global-registrator-20.12.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 4, 2026

Copy link
Copy Markdown
Contributor

Bumps @happy-dom/global-registrator from 20.11.2 to 20.12.0.

Release notes

Sourced from @​happy-dom/global-registrator's releases.

v20.12.0

🎨 Features

  • Adds support for Animation, AnimationTimeline, DocumentTimeline and KeyframeEffect - By @​capricorn86
  • Adds support for Element.animate() - By @​capricorn86

v20.11.15

👷‍♂️ Patch fixes

  • Fixes issue with upgrading custom element caused by previous release - By @​capricorn86 in task #2331

v20.11.14

👷‍♂️ Patch fixes

  • Change prototype of element to the custom element when upgraded to keep the reference - By @​capricorn86 in task #2331
    • The previous solution replaced the element with the custom element and copied it's properties, but this made any reference to the original element invalid

v20.11.13

👷‍♂️ Patch fixes

v20.11.12

👷‍♂️ Patch fixes

v20.11.11

👷‍♂️ Patch fixes

v20.11.10

👷‍♂️ Patch fixes

  • Return default validationMessage for built-in constraint violations - By @​mixelburg in task #2187

v20.11.9

👷‍♂️ Patch fixes

  • Make sure that the element has been upgraded before invoking reaction callback in custom elements - By @​capricorn86 in task #2317

v20.11.8

👷‍♂️ Patch fixes

  • Values pushed or assigned to the adoptedStyleSheet array should be validated - By @​capricorn86 in task #2315

v20.11.7

👷‍♂️ Patch fixes

  • The properties Document.adoptedStyleSheets and ShadowRoot.adoptedStyleSheets should validate it's value - By @​capricorn86 in task #2313

v20.11.6

👷‍♂️ Patch fixes

v20.11.5

👷‍♂️ Patch fixes

  • Allow explicit element types for querySelector (e.g. querySelector<HTMLInputElement>(".my-input")) - By @​cyphercodes

... (truncated)

Commits
  • 4b0af0c feat: Adds support for Element.animate (#2335)
  • e1a8af5 fix: #2331 Fixes issue with upgrading custom element caused by previous rel...
  • 096317b fix: #2331 Change prototype of element to the custom element when upgraded ...
  • 9ff4b47 fix: #2329 Copy event listeners when upgrading a custom element (#2330)
  • 3d282e5 fix: #2322 Add check for if the node is still a parent during disconnect (#...
  • 64e2c77 fix: #2253 Use proxy as parent node in Node.connectedToNode() (#2254)
  • a00ec69 fix: #2319 Removing a child incorrectly calls flow for connecting node (#2321)
  • 152b5d8 fix: #2319 Removing a child incorrectly calls flow for connecting node (#2320)
  • 1100b9c fix: #2187 Return default validationMessage for built-in constraint violati...
  • 1893795 fix: #2317 Make sure that the element has been upgraded before invoking rea...
  • Additional commits viewable in compare view

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [@happy-dom/global-registrator](https://github.com/capricorn86/happy-dom) from 20.11.2 to 20.12.0.
- [Release notes](https://github.com/capricorn86/happy-dom/releases)
- [Commits](capricorn86/happy-dom@v20.11.2...v20.12.0)

---
updated-dependencies:
- dependency-name: "@happy-dom/global-registrator"
  dependency-version: 20.12.0
  dependency-type: direct:development
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update Javascript code labels Sep 4, 2026
@dependabot
dependabot Bot requested a review from caseybaggz as a code owner September 4, 2026 00:53
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update Javascript code labels Sep 4, 2026
@changeset-bot

changeset-bot Bot commented Sep 4, 2026

Copy link
Copy Markdown

⚠️ No Changeset found

Latest commit: 9a6058a

Merging this PR will not cause a version bump for any packages. If these changes should not result in a new version, you're good to go. If these changes should result in a version bump, you need to add a changeset.

This PR includes no changesets

When changesets are added to this PR, you'll see the packages that this PR includes changesets for and the associated semver types

Click here to learn what changesets are, and how to add one.

Click here if you're a maintainer who wants to add a changeset to this PR

@vercel

vercel Bot commented Sep 4, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
cerberus-docs Skipped Skipped Sep 4, 2026 12:53am UTC

Request Review

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update Javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants