Skip to content

feat: support nested array - #3785

Open
SpadeA-Tang wants to merge 1 commit into
milvus-io:masterfrom
SpadeA-Tang:nested-array
Open

feat: support nested array#3785
SpadeA-Tang wants to merge 1 commit into
milvus-io:masterfrom
SpadeA-Tang:nested-array

Conversation

@SpadeA-Tang

Copy link
Copy Markdown
Contributor

No description provided.

@sre-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: SpadeA-Tang
To complete the pull request process, please assign longjiquan after the PR has been reviewed.
You can assign the PR to them by writing /assign @longjiquan in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@mergify

mergify Bot commented Sep 8, 2026

Copy link
Copy Markdown

Tick the box to add this pull request to the merge queue (same as @mergifyio queue).

  • Queue this pull request

@codecov

codecov Bot commented Sep 8, 2026

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 98.20359% with 3 lines in your changes missing coverage. Please review.
✅ Project coverage is 94.30%. Comparing base (a4b3b38) to head (c731fb3).

Files with missing lines Patch % Lines
pymilvus/orm/schema.py 96.77% 2 Missing ⚠️
pymilvus/client/field_data_extractors.py 92.85% 1 Missing ⚠️
Additional details and impacted files
@@            Coverage Diff             @@
##           master    #3785      +/-   ##
==========================================
+ Coverage   94.22%   94.30%   +0.08%     
==========================================
  Files          77       77              
  Lines       16204    16367     +163     
==========================================
+ Hits        15268    15435     +167     
+ Misses        936      932       -4     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@SpadeA-Tang
SpadeA-Tang force-pushed the nested-array branch 2 times, most recently from 8c742b7 to d6ccfe0 Compare September 8, 2026 06:45
@mergify mergify Bot added the ci-passed label Sep 8, 2026
@mergify mergify Bot added ci-passed and removed ci-passed labels Sep 8, 2026
Signed-off-by: SpadeA <tangchenjie1210@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants