Skip to content
This repository was archived by the owner on Mar 21, 2025. It is now read-only.

test: remove nightly-v2_3_x test - #855

Open
ThreadDao wants to merge 1 commit into
milvus-io:masterfrom
ThreadDao:remove-2.3-nightly
Open

test: remove nightly-v2_3_x test#855
ThreadDao wants to merge 1 commit into
milvus-io:masterfrom
ThreadDao:remove-2.3-nightly

Conversation

@ThreadDao

Copy link
Copy Markdown
Contributor

No description provided.

Signed-off-by: ThreadDao <yufen.zong@zilliz.com>
@sre-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ThreadDao
To complete the pull request process, please assign sunby after the PR has been reviewed.
You can assign the PR to them by writing /assign @sunby in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sre-ci-robot sre-ci-robot added the size/M Denotes a PR that changes 30-99 lines. label Dec 9, 2024
@mergify mergify Bot added ci-passed auto merge needed label kind/workflow labels Dec 9, 2024
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

ci-passed auto merge needed label kind/workflow size/M Denotes a PR that changes 30-99 lines.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants