Skip to content
This repository was archived by the owner on Mar 21, 2025. It is now read-only.

test: add dynamic case and upgrade gotestsum - #786

Open
ThreadDao wants to merge 2 commits into
milvus-io:v2.3.xfrom
ThreadDao:v2.3.x-test
Open

test: add dynamic case and upgrade gotestsum#786
ThreadDao wants to merge 2 commits into
milvus-io:v2.3.xfrom
ThreadDao:v2.3.x-test

Conversation

@ThreadDao

Copy link
Copy Markdown
Contributor

No description provided.

@sre-ci-robot
sre-ci-robot requested review from sunby and yah01 July 9, 2024 04:14
@sre-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: ThreadDao
To complete the pull request process, please assign congqixia after the PR has been reviewed.
You can assign the PR to them by writing /assign @congqixia in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sre-ci-robot sre-ci-robot added the size/M Denotes a PR that changes 30-99 lines. label Jul 9, 2024
Signed-off-by: ThreadDao <yufen.zong@zilliz.com>
@ThreadDao
ThreadDao force-pushed the v2.3.x-test branch 3 times, most recently from 87e3dc9 to 2a3b6e9 Compare July 9, 2024 11:47
@ThreadDao ThreadDao changed the title [cherry-pick] test: add dynamic case and upgrade gotestsum test: add dynamic case and upgrade gotestsum Jul 9, 2024
Signed-off-by: ThreadDao <yufen.zong@zilliz.com>
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

size/M Denotes a PR that changes 30-99 lines.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants