Skip to content
This repository was archived by the owner on Mar 21, 2025. It is now read-only.

Fix PKs2Expr - #430

Open
xwjdsh wants to merge 1 commit into
milvus-io:masterfrom
pandodao:feat/fix-pks-to-expr
Open

Fix PKs2Expr#430
xwjdsh wants to merge 1 commit into
milvus-io:masterfrom
pandodao:feat/fix-pks-to-expr

Conversation

@xwjdsh

@xwjdsh xwjdsh commented Mar 27, 2023

Copy link
Copy Markdown

Close #429

Signed-off-by: Wen Sun <iwendellsun@gmail.com>
@sre-ci-robot
sre-ci-robot requested review from congqixia and sunby March 27, 2023 10:49
@sre-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: xwjdsh
To complete the pull request process, please assign sunby after the PR has been reviewed.
You can assign the PR to them by writing /assign @sunby in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@sre-ci-robot sre-ci-robot added the size/M Denotes a PR that changes 30-99 lines. label Mar 27, 2023
@codecov

codecov Bot commented Mar 29, 2023

Copy link
Copy Markdown

Codecov Report

All modified and coverable lines are covered by tests ✅

Project coverage is 87.49%. Comparing base (53227fa) to head (e147ca0).
Report is 217 commits behind head on master.

Additional details and impacted files

Impacted file tree graph

@@           Coverage Diff           @@
##           master     #430   +/-   ##
=======================================
  Coverage   87.49%   87.49%           
=======================================
  Files          30       30           
  Lines        2815     2815           
=======================================
  Hits         2463     2463           
  Misses        267      267           
  Partials       85       85           
Files with missing lines Coverage Δ
client/client_grpc_data.go 72.48% <100.00%> (ø)

@mergify mergify Bot added ci-passed auto merge needed label and removed ci-passed auto merge needed label labels Dec 29, 2023
@mergify mergify Bot added ci-passed auto merge needed label and removed ci-passed auto merge needed label labels Dec 29, 2023
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

size/M Denotes a PR that changes 30-99 lines.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[Bug]: DeleteByPks is disabled if the string PK contains spaces

2 participants