Skip to content

Add joeywan-google to inference-perf-reviewers - #759

Open
joeywan-google wants to merge 1 commit into
kubernetes-sigs:mainfrom
joeywan-google:add-joeywan-google-reviewer
Open

Add joeywan-google to inference-perf-reviewers#759
joeywan-google wants to merge 1 commit into
kubernetes-sigs:mainfrom
joeywan-google:add-joeywan-google-reviewer

Conversation

@joeywan-google

Copy link
Copy Markdown

Adds joeywan-google to the inference-perf-reviewers alias.

Signed-off-by: joeywan-google <257918665+joeywan-google@users.noreply.github.com>
@kubernetes-prow
kubernetes-prow Bot requested review from Bslabe123 and alonh August 24, 2026 19:29
@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: joeywan-google
Once this PR has been reviewed and has the lgtm label, please assign sachinvarghese for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@linux-foundation-easycla

Copy link
Copy Markdown

CLA Not Signed

@github-actions github-actions Bot added the do-not-merge DEPRECATED. Indicates that a PR should not merge. Label can only be manually applied/removed. label Aug 24, 2026
@kubernetes-prow

Copy link
Copy Markdown

Welcome @joeywan-google!

It looks like this is your first PR to kubernetes-sigs/inference-perf 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes-sigs/inference-perf has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@kubernetes-prow kubernetes-prow Bot added cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. labels Aug 24, 2026
@kubernetes-prow

Copy link
Copy Markdown

The OWNERS file contains untrusted users, which makes it INVALID. The following users are mentioned in OWNERS file(s) but are untrusted for the following reasons. One way to make the user trusted is to add them as members of the kubernetes-sigs org. You can then trigger verification by writing /verify-owners in a comment.

  • joeywan-google
    • User is not a member of the org. Satisfy at least one of these conditions to make the user trusted.

@kubernetes-prow kubernetes-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Aug 24, 2026
@Bslabe123

Copy link
Copy Markdown
Contributor

Cross posting: kubernetes/org#6581

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. do-not-merge DEPRECATED. Indicates that a PR should not merge. Label can only be manually applied/removed. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants