Skip to content

[main] Upgrade to latest dependencies - #672

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main
Aug 21, 2026
Merged

[main] Upgrade to latest dependencies#672
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

/cc knative-extensions/kn-plugin-quickstart-approvers
/assign knative-extensions/kn-plugin-quickstart-approvers

Produced by: knative-extensions/knobots/actions/update-deps

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team August 21, 2026 13:52
@knative-prow

knative-prow Bot commented Aug 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign psschwei for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Aug 21, 2026
@knative-prow
knative-prow Bot merged commit 03287fe into knative-extensions:main Aug 21, 2026
23 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant