Skip to content

[main] Upgrade to latest dependencies - #670

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main
Jul 28, 2026
Merged

[main] Upgrade to latest dependencies#670
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

/cc knative-extensions/kn-plugin-quickstart-approvers
/assign knative-extensions/kn-plugin-quickstart-approvers

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/client/pkg 025f9b5...7663461:
  > 7663461 upgrade to latest dependencies (# 2253)
  > 00c078d upgrade to latest dependencies (# 2252)
  > cc108d6 upgrade to latest dependencies (# 2251)
  > ca329bb upgrade to latest dependencies (# 2250)
  > 969d04b upgrade to latest dependencies (# 2249)
  > 11ea2c1 upgrade to latest dependencies (# 2248)
  > 9ce38f8 upgrade to latest dependencies (# 2247)
  > 2370657 upgrade to latest dependencies (# 2246)
  > 1d03beb upgrade to latest dependencies (# 2245)
  > a189866 upgrade to latest dependencies (# 2244)
  > 43f7c74 upgrade to latest dependencies (# 2243)
  > dc5d0a4 upgrade to latest dependencies (# 2242)
  > 6213d96 upgrade to latest dependencies (# 2241)
  > b809afb upgrade to latest dependencies (# 2240)
bumping golang.org/x/sys d58dcfa...9e7e939:
  > 9e7e939 cpu: handle vendor suffixes in parseRelease
  > f6fb8a1 unix: use epoll_pwait rather than epoll_wait
  > f3eeabf windows: avoid length overflow in NewNTString
  > 3cb6647 unix: update glibc to 2.43
  > c507910 windows: document safe usage of TrusteeValue
bumping k8s.io/apimachinery 5e12da0...3e341b5:
  > 3e341b5 Merge pull request # 139666 from kmala/automated-cherry-pick-of-# 137451-upstream-release-1.35
  > 539f8c5 Update google.golang.org/protobuf to v1.36.12-0.20260120151049-f2248ac996af
bumping go.yaml.in/yaml/v3 c3552c1...e16c7af:
  > e16c7af Remove empty go.sum file
  > 6abcd01 Fix unhandled nil-errors in tests
  > 9481db7 Remove suite_test.go and gopkg.in/check.v1
  > 485f5fc Use standard testing library for limit_test.go
  > 3e00046 Use standard testing library for node_test.go
  > f612058 Use standard testing library for decode_test.go
  > 078f958 Use standard testing library encode_test.go
  > 3847ac0 [v3] backport CodeQL workflow
  > 1666663 Revert incorrect gofmt changes
  > 73290cb Retract v3 tags that cannot be installed
bumping golang.org/x/text f4bb632...724af9c:
  > 724af9c go.mod: update golang.org/x dependencies
  > bf5b9d6 internal/export/idna: always treat Punycode encoding pure ASCII as an error
  > b326f3d go.mod: update golang.org/x dependencies
  > 5ae8e57 unicode/norm: avoid infinite loop on invalid input
  > 0dc94a2 all: fix some comments

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team July 28, 2026 02:10
@knative-prow

knative-prow Bot commented Jul 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. label Jul 28, 2026
@knative-prow
knative-prow Bot merged commit bece325 into knative-extensions:main Jul 28, 2026
23 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant