Skip to content

wait for pods to appear before checking ready - #669

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
psschwei:fix-wait-for-create-race
Jul 30, 2026
Merged

wait for pods to appear before checking ready#669
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
psschwei:fix-wait-for-create-race

Conversation

@psschwei

Copy link
Copy Markdown
Member

Changes

  • 🧹 Wait for pods to be created before checking if they are ready. Fixes an issue users were reporting where kourier install was failing. Requires at least kubectl v1.34, so also adds a check in the install step for kubectl version.

/kind performance

Fixes #668

Release Note

Adds a minimum kubectl version check during the install step, in line with the minimum kubernetes version.

Docs


Signed-off-by: Paul S. Schweigert <paul@paulschweigert.com>
@knative-prow
knative-prow Bot requested review from dprotaso and dsimansk July 23, 2026 21:08
@knative-prow knative-prow Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 23, 2026
@dsimansk

Copy link
Copy Markdown
Contributor

/approve
/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 30, 2026
@knative-prow

knative-prow Bot commented Jul 30, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk, psschwei

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow
knative-prow Bot merged commit e19d850 into knative-extensions:main Jul 30, 2026
23 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. kind/performance lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Kourier installation fails on Kind with "no matching resources found" — kubectl wait doesn't wait for pods to appear

2 participants