Skip to content

[release-1.21] Upgrade to latest dependencies - #664

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:release-1.21from
knative-automation:auto-updates/update-deps-release-1.21
May 31, 2026
Merged

[release-1.21] Upgrade to latest dependencies#664
knative-prow[bot] merged 1 commit into
knative-extensions:release-1.21from
knative-automation:auto-updates/update-deps-release-1.21

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

bump -dprotaso

/cc knative-extensions/kn-plugin-quickstart-approvers
/assign knative-extensions/kn-plugin-quickstart-approvers

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/hack bf6758c...c985ed3:
  > c985ed3 [release-1.21] Add env var to pass extra flag to license check (# 471)
  > e37ec67 Bump go-licenses to v2.0.1 release (# 468)

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team May 31, 2026 00:36
@knative-prow

knative-prow Bot commented May 31, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign gauron99 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label May 31, 2026
@knative-prow
knative-prow Bot merged commit db2906f into knative-extensions:release-1.21 May 31, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant