Skip to content

[main] Update community files - #661

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main
May 21, 2026
Merged

[main] Update community files#661
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-community-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

Produced by: knative-extensions/knobots/actions/update-community

Details:


Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow

knative-prow Bot commented May 21, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow

knative-prow Bot commented May 21, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot added do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels May 21, 2026
@dprotaso

Copy link
Copy Markdown
Contributor

/verify-owners

@knative-prow

knative-prow Bot commented May 21, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@knative-prow knative-prow Bot removed the do-not-merge/invalid-owners-file Indicates that a PR should not merge because it has an invalid OWNERS file in it. label May 21, 2026
@knative-prow
knative-prow Bot merged commit 8109f46 into knative-extensions:main May 21, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants