Skip to content

[main] Upgrade to latest dependencies - #656

Merged
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main
Apr 28, 2026
Merged

[main] Upgrade to latest dependencies#656
knative-prow[bot] merged 1 commit into
knative-extensions:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

release -dprotaso

/cc knative-extensions/kn-plugin-quickstart-approvers
/assign knative-extensions/kn-plugin-quickstart-approvers

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/client/pkg 4b82b12...f20b8b8:
  > f20b8b8 upgrade to latest dependencies (# 2204)
  > 0f8d505 upgrade to latest dependencies (# 2203)
  > 21f1fb5 Update community files (# 2202)

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team April 28, 2026 21:44
@knative-prow

knative-prow Bot commented Apr 28, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign psschwei for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Apr 28, 2026
@knative-prow
knative-prow Bot merged commit 6dba603 into knative-extensions:main Apr 28, 2026
18 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant