[main] Upgrade to latest dependencies - #654
Merged
knative-prow[bot] merged 1 commit intoApr 27, 2026
Merged
Conversation
bumping knative.dev/client/pkg 2ae9a8e...4b82b12: > 4b82b12 upgrade to latest dependencies (# 2201) > 0953acf upgrade to latest dependencies (# 2200) > 512ade3 upgrade to latest dependencies (# 2199) > bbe46a8 upgrade to latest dependencies (# 2198) > b67c724 upgrade to latest dependencies (# 2197) > e9d4a87 upgrade to latest dependencies (# 2196) > 252b869 upgrade to latest dependencies (# 2195) Signed-off-by: Knative Automation <automation@knative.team>
knative-automation
force-pushed
the
auto-updates/update-deps-main
branch
from
April 27, 2026 14:43
8969595 to
cbf68ce
Compare
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: knative-automation The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
bump -dprotaso
/cc knative-extensions/kn-plugin-quickstart-approvers
/assign knative-extensions/kn-plugin-quickstart-approvers
Produced by: knative-extensions/knobots/actions/update-deps