build(deps): bump arduino/setup-protoc from 1.3.0 to 3.0.0 - #507
build(deps): bump arduino/setup-protoc from 1.3.0 to 3.0.0#507dependabot[bot] wants to merge 1 commit into
Conversation
Bumps [arduino/setup-protoc](https://github.com/arduino/setup-protoc) from 1.3.0 to 3.0.0. - [Release notes](https://github.com/arduino/setup-protoc/releases) - [Commits](arduino/setup-protoc@149f6c8...c65c819) --- updated-dependencies: - dependency-name: arduino/setup-protoc dependency-version: 3.0.0 dependency-type: direct:production update-type: version-update:semver-major ... Signed-off-by: dependabot[bot] <support@github.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a etcd-io member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Bumps arduino/setup-protoc from 1.3.0 to 3.0.0.
Release notes
Sourced from arduino/setup-protoc's releases.
Commits
c65c819Upgrade to node 20 (#95)52a53b4Merge pull request #93 from arduino/dependabot/npm_and_yarn/babel/traverse-7....cf7ab7fBump@babel/traversefrom 7.22.1 to 7.23.2e2995baCorrectconvetiontypo in README (#91)a8b67babump semver to 7.5.3 (#90)1530d62Bump semver from 7.5.1 to 7.5.2 (#87)0fbeb49Exposepathandversioninoutputs(#89)9b1ee5bv2 release note (#82)28fd3e5Support only the new protobuf versioning scheme (#78)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)