Skip to content

[2026-09-07] Manual Dependency Bump - #22407

Merged
fuweid merged 3 commits into
etcd-io:mainfrom
ivanvc:20260907-dependency-bump
Sep 10, 2026
Merged

[2026-09-07] Manual Dependency Bump#22407
fuweid merged 3 commits into
etcd-io:mainfrom
ivanvc:20260907-dependency-bump

Conversation

@codecov

codecov Bot commented Sep 10, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 72.97%. Comparing base (6799d48) to head (fd0c207).

Additional details and impacted files

see 24 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #22407      +/-   ##
==========================================
- Coverage   73.01%   72.97%   -0.04%     
==========================================
  Files         448      448              
  Lines       31571    31571              
==========================================
- Hits        23050    23038      -12     
- Misses       8518     8530      +12     
  Partials        3        3              

Continue to review full report in Codecov by Harness.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 6799d48...fd0c207. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@ivanvc
ivanvc force-pushed the 20260907-dependency-bump branch from 2ae72d8 to 6cac3ef Compare September 10, 2026 02:20
Comment thread scripts/fix/dep.sh Outdated
function select_latest_version {
local dep="$1"
local all_deps="$2"
echo "${all_deps}" | grep "^${dep}," | cut -d ',' -f 2 | sort -uV | tail -1

@fuweid fuweid Sep 10, 2026

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I don't think we should use rc version, right? If we use, sort is not good enough to handle that

$ printf '%s\n' v1.2.3 v1.2.3-rc.1 | sort -uV
v1.2.3
v1.2.3-rc.1

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Crap, this file shouldn't be part of this PR. I use it to fix version mismatches (#22224).

I'll remove this file.

But to answer your question: I haven't seen a Go dependency use prereleases.

Comment thread scripts/fix/dep.sh Outdated
References: etcd-io#22347

Signed-off-by: Ivan Valdes <ivan@vald.es>
… 2.13.2

References:
* etcd-io#22360
* etcd-io#22363

Signed-off-by: Ivan Valdes <ivan@vald.es>
Also bump
go.opentelemetry.io/contrib/instrumentation/google.golang.org/grpc/otelgrpc
from 0.70.0 to 0.71.0.

References:
* etcd-io#22403
* etcd-io#22402
* etcd-io#22374
* etcd-io#22371
* etcd-io#22362
* etcd-io#22361

Signed-off-by: Ivan Valdes <ivan@vald.es>
@ivanvc
ivanvc force-pushed the 20260907-dependency-bump branch from 6cac3ef to fd0c207 Compare September 10, 2026 18:49
@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: fuweid, ivanvc

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@fuweid
fuweid merged commit e703769 into etcd-io:main Sep 10, 2026
33 checks passed
@ivanvc
ivanvc deleted the 20260907-dependency-bump branch September 10, 2026 21:29
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

2 participants