Skip to content

Update security reporting info. - #22308

Merged
ahrtr merged 1 commit into
etcd-io:mainfrom
jberkus:newsecurity
Aug 24, 2026
Merged

Update security reporting info.#22308
ahrtr merged 1 commit into
etcd-io:mainfrom
jberkus:newsecurity

Conversation

@jberkus

@jberkus jberkus commented Aug 20, 2026

Copy link
Copy Markdown
Contributor

All mentions of security reporting have been updated to indicate use of the GitHub vuln report form, and not the mailing address.

@serathius @fuweid @ahrtr

All mentions of security reporting have been updated to indicate use of the GitHub vuln report form, and not the mailing address.

Signed-off-by: Josh Berkus <josh@agliodbs.com>
@codecov

codecov Bot commented Aug 20, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 69.74%. Comparing base (7b2ad21) to head (6afd659).
⚠️ Report is 5 commits behind head on main.

Additional details and impacted files

see 19 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #22308      +/-   ##
==========================================
+ Coverage   69.64%   69.74%   +0.09%     
==========================================
  Files         449      449              
  Lines       38196    38196              
==========================================
+ Hits        26602    26640      +38     
+ Misses      10163    10125      -38     
  Partials     1431     1431              

Continue to review full report in Codecov by Harness.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 7b2ad21...6afd659. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@jberkus

jberkus commented Aug 20, 2026

Copy link
Copy Markdown
Contributor Author

So, the test failure is unrelated to this PR. Flaky test?

@fuweid

fuweid commented Aug 20, 2026

Copy link
Copy Markdown
Member

/retest

@ahrtr
ahrtr merged commit ca5bf0f into etcd-io:main Aug 24, 2026
35 checks passed
@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ahrtr, fuweid, jberkus, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [ahrtr,fuweid,serathius]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

4 participants