Skip to content

Run dependabot on Mondays - #22300

Merged
ahrtr merged 1 commit into
etcd-io:mainfrom
ivanvc:dependabot-config-monday-schedule
Aug 27, 2026
Merged

Run dependabot on Mondays#22300
ahrtr merged 1 commit into
etcd-io:mainfrom
ivanvc:dependabot-config-monday-schedule

Conversation

@ivanvc

@ivanvc ivanvc commented Aug 17, 2026

Copy link
Copy Markdown
Member

Recently we've been having Dependabot (not security) updates on Wednesdays and Thursdays. This breaks our manual weekly dependency. Although Monday is the default day
(https://docs.github.com/en/code-security/reference/supply-chain-security/dependabot-options-reference#interval), explicitly define is as Monday.

Recently we've been having Dependabot (not security) updates on
Wednesdays and Thursdays. This breaks our manual weekly dependency.
Although Monday is the default day
(https://docs.github.com/en/code-security/reference/supply-chain-security/dependabot-options-reference#interval),
explicitly define is as Monday.

Signed-off-by: Ivan Valdes <ivan@vald.es>
@kubernetes-prow

Copy link
Copy Markdown

@ivanvc: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
pull-etcd-robustness-amd64 4555000 link true /test pull-etcd-robustness-amd64

Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@codecov

codecov Bot commented Aug 17, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 69.64%. Comparing base (00a7653) to head (4555000).

Additional details and impacted files

see 14 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #22300      +/-   ##
==========================================
- Coverage   69.66%   69.64%   -0.03%     
==========================================
  Files         449      449              
  Lines       38193    38193              
==========================================
- Hits        26609    26600       -9     
- Misses      10158    10169      +11     
+ Partials     1426     1424       -2     

Continue to review full report in Codecov by Harness.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 00a7653...4555000. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ahrtr, ivanvc

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ahrtr
ahrtr merged commit 1cca2c9 into etcd-io:main Aug 27, 2026
33 of 34 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

2 participants