Skip to content

build(deps): bump distroless/static-debian12 from a9fcaed to 6447365 - #22273

Merged
ivanvc merged 1 commit into
mainfrom
dependabot/docker/distroless/static-debian12-6447365a6337c3732f412d1b74357b30a633831955b2bc45552b0086be907687
Aug 18, 2026
Merged

build(deps): bump distroless/static-debian12 from a9fcaed to 6447365#22273
ivanvc merged 1 commit into
mainfrom
dependabot/docker/distroless/static-debian12-6447365a6337c3732f412d1b74357b30a633831955b2bc45552b0086be907687

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 13, 2026

Copy link
Copy Markdown
Contributor

Bumps distroless/static-debian12 from a9fcaed to 6447365.

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file docker Pull requests that update Docker code labels Aug 13, 2026
@kubernetes-prow

Copy link
Copy Markdown

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a etcd-io member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@ivanvc

ivanvc commented Aug 14, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@codecov

codecov Bot commented Aug 14, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 69.67%. Comparing base (00a7653) to head (c886e87).
⚠️ Report is 11 commits behind head on main.

Additional details and impacted files

see 20 files with indirect coverage changes

@@           Coverage Diff           @@
##             main   #22273   +/-   ##
=======================================
  Coverage   69.66%   69.67%           
=======================================
  Files         449      449           
  Lines       38193    38193           
=======================================
+ Hits        26609    26611    +2     
- Misses      10158    10161    +3     
+ Partials     1426     1421    -5     

Continue to review full report in Codecov by Harness.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 00a7653...c886e87. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@ivanvc

ivanvc commented Aug 18, 2026

Copy link
Copy Markdown
Member

@dependabot rebase

Bumps distroless/static-debian12 from `a9fcaed` to `6447365`.

---
updated-dependencies:
- dependency-name: distroless/static-debian12
  dependency-version: 6447365a6337c3732f412d1b74357b30a633831955b2bc45552b0086be907687
  dependency-type: direct:production
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/docker/distroless/static-debian12-6447365a6337c3732f412d1b74357b30a633831955b2bc45552b0086be907687 branch from fa7f836 to c886e87 Compare August 18, 2026 18:15
@ivanvc

ivanvc commented Aug 18, 2026

Copy link
Copy Markdown
Member

/retest

@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], ivanvc

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ivanvc
ivanvc merged commit 80dfede into main Aug 18, 2026
34 checks passed
@dependabot
dependabot Bot deleted the dependabot/docker/distroless/static-debian12-6447365a6337c3732f412d1b74357b30a633831955b2bc45552b0086be907687 branch August 18, 2026 21:15
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved dependencies Pull requests that update a dependency file docker Pull requests that update Docker code ok-to-test size/XS

Development

Successfully merging this pull request may close these issues.

1 participant