Stop purging v2 snapshot files - #22271
Conversation
Signed-off-by: Benjamin Wang <benjamin.ahrtr@gmail.com>
|
/test pull-etcd-contrib-mixin |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files
... and 18 files with indirect coverage changes @@ Coverage Diff @@
## main #22271 +/- ##
==========================================
- Coverage 69.71% 69.66% -0.06%
==========================================
Files 449 449
Lines 38187 38182 -5
==========================================
- Hits 26622 26599 -23
- Misses 10141 10151 +10
- Partials 1424 1432 +8 Continue to review full report in Codecov by Harness.
🚀 New features to boost your workflow:
|
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: ahrtr, fuweid The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
part of #20187
We don't generate v2 snapshot files anymore, so no need to purge v2 snapshot files.