Skip to content

clean up gogo: drop nullable=false from etcdserverpb package - #21292

Merged
serathius merged 3 commits into
etcd-io:mainfrom
liggitt:protoc-etcdserverpb
Feb 12, 2026
Merged

clean up gogo: drop nullable=false from etcdserverpb package#21292
serathius merged 3 commits into
etcd-io:mainfrom
liggitt:protoc-etcdserverpb

Conversation

@liggitt

@liggitt liggitt commented Feb 11, 2026

Copy link
Copy Markdown
Contributor

xref #14533 (comment)

splits out dropping gogoproto.nullable=false option from etcdserverpb package, switches in-tree callers that I could find to use the nil-safe getters

@k8s-ci-robot

Copy link
Copy Markdown

Hi @liggitt. Thanks for your PR.

I'm waiting for a etcd-io member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Comment thread api/etcdserverpb/etcdserver.pb.go
Comment thread api/etcdserverpb/etcdserver.proto
@ahrtr

ahrtr commented Feb 11, 2026

Copy link
Copy Markdown
Member

/ok-to-test

@codecov

codecov Bot commented Feb 11, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 68.46%. Comparing base (6b93faf) to head (8956ed7).
⚠️ Report is 6 commits behind head on main.

Additional details and impacted files
Files with missing lines Coverage Δ
etcdutl/snapshot/v3_snapshot.go 58.30% <100.00%> (+12.54%) ⬆️
server/etcdserver/bootstrap.go 66.52% <100.00%> (ø)

... and 45 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #21292      +/-   ##
==========================================
- Coverage   68.99%   68.46%   -0.53%     
==========================================
  Files         410      428      +18     
  Lines       34606    35256     +650     
==========================================
+ Hits        23876    24138     +262     
- Misses       9343     9729     +386     
- Partials     1387     1389       +2     

Continue to review full report in Codecov by Sentry.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 6b93faf...8956ed7. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@liggitt

liggitt commented Feb 11, 2026

Copy link
Copy Markdown
Contributor Author

@ahrtr this is ready

@liggitt

liggitt commented Feb 11, 2026

Copy link
Copy Markdown
Contributor Author

/retest

Signed-off-by: Jordan Liggitt <liggitt@google.com>
Signed-off-by: Jordan Liggitt <liggitt@google.com>
@liggitt
liggitt force-pushed the protoc-etcdserverpb branch from 7d44e6d to 88fbd40 Compare February 11, 2026 19:14
@ahrtr

ahrtr commented Feb 11, 2026

Copy link
Copy Markdown
Member

pls fix the workflow failure

@liggitt

liggitt commented Feb 11, 2026

Copy link
Copy Markdown
Contributor Author

pls fix the workflow failure

uh... it's complaining about something that is not accurate ...

../etcdutl/snapshot/v3_snapshot.go:522:2: SA4006: this value of m is never used (staticcheck)
	m := s.cl.MemberByName(s.name)

but I didn't touch that line in this PR ... I'm guessing staticcheck doesn't understand new making use of a field

        m := s.cl.MemberByName(s.name)
-       md := &etcdserverpb.Metadata{NodeID: uint64(m.ID), ClusterID: uint64(s.cl.ID())}
+       md := &etcdserverpb.Metadata{NodeID: new(uint64(m.ID)), ClusterID: new(uint64(s.cl.ID()))}
        metadata, merr := md.Marshal()

Signed-off-by: Jordan Liggitt <liggitt@google.com>
@liggitt
liggitt force-pushed the protoc-etcdserverpb branch from 88fbd40 to 8956ed7 Compare February 11, 2026 20:10
@liggitt

liggitt commented Feb 11, 2026

Copy link
Copy Markdown
Contributor Author

yup, dominikh/go-tools#1698 was the issue, added a nolint to that line for now

@ahrtr ahrtr left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM & thx

cc. @fuweid

@k8s-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ahrtr, liggitt, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serathius
serathius merged commit 827d9e7 into etcd-io:main Feb 12, 2026
32 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

4 participants