Skip to content

Document how to debug performance of Prow jobs in etcd - #20796

Merged
serathius merged 1 commit into
etcd-io:mainfrom
ronaldngounou:debug-performance-prow
Dec 19, 2025
Merged

Document how to debug performance of Prow jobs in etcd#20796
serathius merged 1 commit into
etcd-io:mainfrom
ronaldngounou:debug-performance-prow

Conversation

@ronaldngounou

Copy link
Copy Markdown
Member

Overview

This PR documents how etcd contributors can navigate performance of prow jobs.
Contributes to issue #18707

@k8s-ci-robot

Copy link
Copy Markdown

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@codecov

codecov Bot commented Oct 19, 2025

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 69.27%. Comparing base (1b2ed5a) to head (417ce15).
⚠️ Report is 127 commits behind head on main.

Additional details and impacted files

see 30 files with indirect coverage changes

@@            Coverage Diff             @@
##             main   #20796      +/-   ##
==========================================
+ Coverage   69.21%   69.27%   +0.06%     
==========================================
  Files         422      422              
  Lines       34831    34841      +10     
==========================================
+ Hits        24108    24137      +29     
+ Misses       9327     9308      -19     
  Partials     1396     1396              

Continue to review full report in Codecov by Sentry.

Legend - Click here to learn more
Δ = absolute <relative> (impact), ø = not affected, ? = missing data
Powered by Codecov. Last update 1b2ed5a...417ce15. Read the comment docs.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@ronaldngounou

ronaldngounou commented Oct 22, 2025

Copy link
Copy Markdown
Member Author

Hi @ivanvc , could you please review?
Is there anything I could do to resolve the failing tests ?

@serathius

Copy link
Copy Markdown
Member

Is there anything I could do to resolve thr failing tests ?

Please rebase

@ivanvc

ivanvc commented Oct 23, 2025

Copy link
Copy Markdown
Member

Hi @ronaldngounou, there are a couple of broken links in the document you pushed. Refer to the Prow job failure: https://prow.k8s.io/view/gs/kubernetes-ci-logs/pr-logs/pull/etcd-io_etcd/20796/pull-etcd-verify/1979810317445632000#

Found broken reference (Prow -> ) at /home/prow/go/src/github.com/etcd-io/etcd/Documentation/contributor-guide/prow_jobs.md:5
Found broken path (dashboards -> monitoring-gke.prow.k8s.io/d/96Q8oOOZk/builds?orgId=1&refresh=30s&var-org=etcd-io&var-repo=etcd&var-build=All&from=now-7d&to=now) at /home/prow/go/src/github.com/etcd-io/etcd/Documentation/contributor-guide/prow_jobs.md:41

@ronaldngounou
ronaldngounou force-pushed the debug-performance-prow branch from eb48c66 to 5598b5d Compare October 25, 2025 02:58
@ronaldngounou

Copy link
Copy Markdown
Member Author

I have rebased and fixed the broken links.
cc @ivanvc @serathius

@ivanvc ivanvc left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks, @ronaldngounou. This is a great start. I left some comments. Please take a look at them.

Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
@ronaldngounou

ronaldngounou commented Oct 29, 2025

Copy link
Copy Markdown
Member Author

Feedback is addressed

@jberkus

jberkus commented Oct 29, 2025

Copy link
Copy Markdown
Contributor

/assign

@ronaldngounou
ronaldngounou force-pushed the debug-performance-prow branch 3 times, most recently from 261530a to 56f9a51 Compare October 31, 2025 07:40
@ronaldngounou

Copy link
Copy Markdown
Member Author

/retest

@ivanvc ivanvc left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the updates, @ronaldngounou. I left some comments after the update.

Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md
@ronaldngounou
ronaldngounou force-pushed the debug-performance-prow branch from 56f9a51 to e6a773b Compare November 1, 2025 01:20
@ronaldngounou

Copy link
Copy Markdown
Member Author

Feedback is addressed @ivanvc

@ronaldngounou
ronaldngounou requested a review from ivanvc December 5, 2025 07:09
@ronaldngounou

ronaldngounou commented Dec 5, 2025

Copy link
Copy Markdown
Member Author

Bump for review @ivanvc @serathius please

@serathius

Copy link
Copy Markdown
Member

@jberkus can you review?

@jberkus

jberkus commented Dec 5, 2025

Copy link
Copy Markdown
Contributor

@serathius I'm assuming the technical info is correct then? I'll do a copyedit, thanks!

@jberkus jberkus left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Per feedback.

Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md
Comment thread Documentation/contributor-guide/prow_jobs.md
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
Comment thread Documentation/contributor-guide/prow_jobs.md
Comment thread Documentation/contributor-guide/prow_jobs.md Outdated
- Introduction to Prow
- How Prow is used for etcd testing
- Navigating perfomance dashboards (Grafana)
- Prow job categories
- Interpreting metrics

Signed-off-by: ronaldngounou <ronald.ngounou@yahoo.com>
@jberkus

jberkus commented Dec 17, 2025

Copy link
Copy Markdown
Contributor

/retest

@jberkus

jberkus commented Dec 18, 2025

Copy link
Copy Markdown
Contributor

OK, looks good to me now. @serathius @ivanvc you can merge.

@k8s-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jberkus, ronaldngounou, serathius

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serathius
serathius merged commit c80e55e into etcd-io:main Dec 19, 2025
32 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Development

Successfully merging this pull request may close these issues.

5 participants