Skip to content

fix: sort owned pods by numeric ordinal instead of name - #496

Merged
ahrtr merged 1 commit into
etcd-io:mainfrom
silentred:fix/list-owned-pods-sort-by-ordinal
Sep 2, 2026
Merged

fix: sort owned pods by numeric ordinal instead of name#496
ahrtr merged 1 commit into
etcd-io:mainfrom
silentred:fix/list-owned-pods-sort-by-ordinal

Conversation

@silentred

Copy link
Copy Markdown
Contributor

/cc @ahrtr @nwnt
Link to #488

Signed-off-by: shenmu.wy <shenmu.wy@antfin.com>
@ahrtr
ahrtr merged commit 38e6cb5 into etcd-io:main Sep 2, 2026
5 checks passed
@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ahrtr, silentred

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants