Skip to content

feat(my-view): add quotes-list-view model - #945

Merged
dogayuksel merged 5 commits into
mainfrom
dy-add-quotes-list-view-model
Aug 14, 2025
Merged

dogayuksel merged 5 commits into
mainfrom
dy-add-quotes-list-view-model

Conversation

@dogayuksel

Copy link
Copy Markdown
Contributor

Description

Adds missing QuotesListView model.

@changeset-bot

changeset-bot Bot commented Aug 12, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0e8222e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@commercetools/composable-commerce-test-data Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@dogayuksel
dogayuksel force-pushed the dy-add-quotes-list-view-model branch from ba0b5ab to 7fa1c2a Compare August 13, 2025 11:01
@dogayuksel dogayuksel self-assigned this Aug 13, 2025
@dogayuksel
dogayuksel marked this pull request as ready for review August 13, 2025 13:09
@dogayuksel
dogayuksel requested review from a team as code owners August 13, 2025 13:09
@dogayuksel
dogayuksel requested a review from a team August 13, 2025 13:09
@dogayuksel dogayuksel added the fe-chapter-rotation Tasks coming from frontend chapter work label Aug 13, 2025

@CarlosCortizasCT CarlosCortizasCT left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice and clean 🙇

@CarlosCortizasCT
CarlosCortizasCT requested a review from a team August 13, 2025 14:35
Comment thread .changeset/wise-feet-jog.md Outdated
Comment thread .changeset/wise-feet-jog.md Outdated
Comment thread .changeset/wise-feet-jog.md
dogayuksel and others added 3 commits August 14, 2025 10:32
Co-authored-by: Ahmed Mehri <ahmed.mehri@commercetools.com>
Co-authored-by: Ahmed Mehri <ahmed.mehri@commercetools.com>
Co-authored-by: Ahmed Mehri <ahmed.mehri@commercetools.com>
@dogayuksel
dogayuksel merged commit 888ba63 into main Aug 14, 2025
4 checks passed
@dogayuksel
dogayuksel deleted the dy-add-quotes-list-view-model branch August 14, 2025 08:43
@ct-changesets ct-changesets Bot mentioned this pull request Aug 7, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

fe-chapter-rotation Tasks coming from frontend chapter work

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants