Skip to content

refactor: to use trusted publishing - #127

Merged
tdeekens merged 2 commits into
mainfrom
td/trusted-publishing
Dec 15, 2025
Merged

refactor: to use trusted publishing#127
tdeekens merged 2 commits into
mainfrom
td/trusted-publishing

Conversation

@tdeekens

Copy link
Copy Markdown
Contributor

No description provided.

@tdeekens
tdeekens requested a review from a team as a code owner December 15, 2025 15:44
@changeset-bot

changeset-bot Bot commented Dec 15, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3470322

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@commercetools/express-request-correlator Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@tdeekens
tdeekens merged commit e1e964d into main Dec 15, 2025
6 checks passed
@tdeekens
tdeekens deleted the td/trusted-publishing branch December 15, 2025 15:45
@ct-changesets ct-changesets Bot mentioned this pull request Dec 15, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant