Skip to content
This repository was archived by the owner on Mar 17, 2026. It is now read-only.

chore: update versions to 24.x to match prod deploy node ver - #2909

Merged
xtbase merged 1 commit into
masterfrom
tea/update-vers
Jan 15, 2026
Merged

chore: update versions to 24.x to match prod deploy node ver#2909
xtbase merged 1 commit into
masterfrom
tea/update-vers

Conversation

@xtbase

@xtbase xtbase commented Jan 15, 2026

Copy link
Copy Markdown
Contributor

What changed? Why?

update to 24.x

Notes to reviewers

How has it been tested?

Have you tested the following pages?

BaseWeb

  • [] base.org
  • [] base.org/names
  • [] base.org/builders
  • [] base.org/ecosystem
  • [] base.org/name/jesse
  • [] base.org/manage-names
  • [] base.org/resources

@cb-heimdall

cb-heimdall commented Jan 15, 2026

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@vercel

vercel Bot commented Jan 15, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Review Updated (UTC)
base-web Ready Ready Preview, Comment Jan 15, 2026 0:39am

@xtbase
xtbase enabled auto-merge (squash) January 15, 2026 00:35
@xtbase
xtbase merged commit ee9a5a1 into master Jan 15, 2026
11 checks passed
@xtbase
xtbase deleted the tea/update-vers branch January 15, 2026 00:39
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants