Skip to content
This repository was archived by the owner on Mar 17, 2026. It is now read-only.

adding Rift to Base ecosystem - #2846

Merged
hughescoin merged 3 commits into
base:masterfrom
vic-stick:add-rift
Dec 11, 2025
Merged

adding Rift to Base ecosystem#2846
hughescoin merged 3 commits into
base:masterfrom
vic-stick:add-rift

Conversation

@vic-stick

Copy link
Copy Markdown
Contributor

What changed? Why?

Added Rift to the Base ecosystem.

Notes to reviewers

How has it been tested?

Ran:
yarn workspace @app/web dev
and
yarn workspace @app/web build

Have you tested the following pages?

BaseWeb

  • [] base.org
  • [] base.org/names
  • [] base.org/builders
  • [] base.org/ecosystem
  • [] base.org/name/jesse
  • [] base.org/manage-names
  • [] base.org/resources

Yes, I ran build and verified.

@vercel

vercel Bot commented Dec 9, 2025

Copy link
Copy Markdown

@vic-stick is attempting to deploy a commit to the Coinbase Team on Vercel.

A member of the Team first needs to authorize it.

@cb-heimdall

cb-heimdall commented Dec 9, 2025

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@roethke
roethke self-requested a review December 10, 2025 23:18
@cb-heimdall

Copy link
Copy Markdown
Collaborator

Review Error for roethke @ 2025-12-10 23:18:50 UTC
User cannot review their own commit

@roethke
roethke requested a review from hughescoin December 10, 2025 23:19
@hughescoin
hughescoin merged commit e0a5300 into base:master Dec 11, 2025
6 of 7 checks passed
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants