Skip to content

feat: add server-core and server-hono - #514

Merged
omeraplak merged 16 commits into
nextfrom
feat/decouple-server
Aug 28, 2025
Merged

omeraplak merged 16 commits into
nextfrom
feat/decouple-server

Conversation

@omeraplak

Copy link
Copy Markdown
Member

PR Checklist

Please check if your PR fulfills the following requirements:

Bugs / Features

What is the current behavior?

What is the new behavior?

fixes (issue)

Notes for reviewers

@cloudflare-workers-and-pages

cloudflare-workers-and-pages Bot commented Aug 23, 2025 •

Copy link
Copy Markdown

Deploying voltagent with  Cloudflare Pages  Cloudflare Pages

Latest commit: dec5f73
Status: ✅  Deploy successful!
Preview URL: https://1c648380.voltagent.pages.dev
Branch Preview URL: https://feat-decouple-server.voltagent.pages.dev

View logs

@changeset-bot

changeset-bot Bot commented Aug 23, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dec5f73

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@omeraplak
omeraplak marked this pull request as ready for review August 26, 2025 21:48
@ghost

ghost commented Aug 26, 2025 •

Copy link
Copy Markdown

📝 Documentation Analysis

Joggr is analyzing the pull for changes that may impact docs, keep calm & carry on...

View previous analysis

⌛️ Analyzing commit: dec5f73 | Powered by Joggr

@omeraplak
omeraplak merged commit e86cadb into next Aug 28, 2025
22 of 23 checks passed
@omeraplak
omeraplak deleted the feat/decouple-server branch August 28, 2025 03:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant