Skip to content

Update CODEOWNERS - #2

Open
itzarjuna1 wants to merge 1 commit into
TheTeamAlexa:mainfrom
itzarjuna1:patch-1
Open

Update CODEOWNERS#2
itzarjuna1 wants to merge 1 commit into
TheTeamAlexa:mainfrom
itzarjuna1:patch-1

Conversation

@itzarjuna1

Copy link
Copy Markdown

No description provided.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant