Skip to content

gasket: fix build with kernel >= 6.13 - #384623

Merged
drupol merged 1 commit into
NixOS:masterfrom
iamruinous:patch-gasket-driver-kernel-6-13
Jun 20, 2025
Merged

drupol merged 1 commit into
NixOS:masterfrom
iamruinous:patch-gasket-driver-kernel-6-13

Conversation

@iamruinous

@iamruinous iamruinous commented Feb 23, 2025 •

Copy link
Copy Markdown
Contributor

Add patch to gasket-driver that fixes the build on kernel 6.13

google/gasket-driver#39
google/gasket-driver#40

Things done

  • Built on platform(s)
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • For non-Linux: Is sandboxing enabled in nix.conf? (See Nix manual)
    • sandbox = relaxed
    • sandbox = true
  • Tested, as applicable:
  • Tested compilation of all packages that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage
  • Tested basic functionality of all binary files (usually in ./result/bin/)
  • 25.05 Release Notes (or backporting 24.11 and 25.05 Release notes)
    • (Package updates) Added a release notes entry if the change is major or breaking
    • (Module updates) Added a release notes entry if the change is significant
    • (Module addition) Added a release notes entry if adding a new NixOS module
  • Fits CONTRIBUTING.md.

Add a 👍 reaction to pull requests you find important.

@github-actions github-actions Bot added 10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. labels Feb 23, 2025
@nix-owners
nix-owners Bot requested a review from kylehendricks February 23, 2025 23:09
@NixOSInfra NixOSInfra added the 12.first-time contribution This PR is the author's first one; please be gentle! label Feb 23, 2025
@wegank wegank added 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. 12.approvals: 1 This PR was reviewed and approved by one person. labels Feb 24, 2025

@drupol drupol left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Hello,

Thanks for this first PR against the Nix project, welcome to the nix community !

Here's my feedback:

  1. Please read the contributing guide to learn how to make a proper commit message. It should look like something: gasket: fix build with kernel >= 6.13
  2. Your PR title must match the commit message

Feel free to ping me if you need some help.

@FliegendeWurst
FliegendeWurst marked this pull request as draft March 23, 2025 10:08
@FliegendeWurst FliegendeWurst added the 6.topic: kernel The Linux kernel label Mar 23, 2025
@iamruinous iamruinous changed the title Add patch to fix gasket build on 6.13 gasket: fix build with kernel >= 6.13 Apr 15, 2025
@github-actions github-actions Bot removed the 6.topic: kernel The Linux kernel label Apr 15, 2025
@iamruinous
iamruinous force-pushed the patch-gasket-driver-kernel-6-13 branch from 7b48f0a to 6f6a4b7 Compare April 15, 2025 04:32
@iamruinous
iamruinous marked this pull request as ready for review April 15, 2025 04:40

@eljamm eljamm left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nixpkgs-review result

Generated using nixpkgs-review-gha

Command: nixpkgs-review pr 384623

Logs: https://github.com/eljamm/nixpkgs-review-gha/actions/runs/15231866943


x86_64-linux

⏩ 6 packages marked as broken and skipped:
  • linuxKernel.packages.linux_5_10.gasket
  • linuxKernel.packages.linux_5_10_hardened.gasket
  • linuxKernel.packages.linux_5_4.gasket
  • linuxKernel.packages.linux_5_4_hardened.gasket
  • linuxPackages_5_10_hardened.gasket
  • linuxPackages_5_4_hardened.gasket
✅ 18 packages built:
  • linuxKernel.packages.linux_5_15.gasket
  • linuxPackages_5_15_hardened.gasket (linuxKernel.packages.linux_5_15_hardened.gasket)
  • linuxKernel.packages.linux_6_1.gasket
  • linuxPackages.gasket (linuxKernel.packages.linux_6_12.gasket)
  • linuxPackages_hardened.gasket (linuxPackages_6_12_hardened.gasket)
  • linuxKernel.packages.linux_6_13.gasket
  • linuxPackages_6_13_hardened.gasket (linuxKernel.packages.linux_6_13_hardened.gasket)
  • linuxPackages_latest.gasket (linuxKernel.packages.linux_6_14.gasket)
  • linuxPackages_6_1_hardened.gasket (linuxKernel.packages.linux_6_1_hardened.gasket)
  • linuxKernel.packages.linux_6_6.gasket
  • linuxPackages_6_6_hardened.gasket (linuxKernel.packages.linux_6_6_hardened.gasket)
  • linuxPackages_ham.gasket (linuxKernel.packages.linux_ham.gasket)
  • linuxPackages_latest-libre.gasket (linuxKernel.packages.linux_latest_libre.gasket)
  • linuxPackages-libre.gasket (linuxKernel.packages.linux_libre.gasket)
  • linuxPackages_lqx.gasket (linuxKernel.packages.linux_lqx.gasket)
  • linuxPackages_xanmod.gasket (linuxKernel.packages.linux_xanmod.gasket)
  • linuxPackages_xanmod_latest.gasket (linuxKernel.packages.linux_xanmod_latest.gasket, linuxPackages_xanmod_stable.gasket)
  • linuxPackages_zen.gasket (linuxKernel.packages.linux_zen.gasket)

aarch64-linux

⏩ 16 packages marked as broken and skipped:
  • linuxKernel.packages.linux_5_10.gasket
  • linuxKernel.packages.linux_5_10_hardened.gasket
  • linuxKernel.packages.linux_5_4.gasket
  • linuxKernel.packages.linux_5_4_hardened.gasket
  • linuxKernel.packages.linux_lqx.gasket
  • linuxKernel.packages.linux_xanmod.gasket
  • linuxKernel.packages.linux_xanmod_latest.gasket
  • linuxKernel.packages.linux_xanmod_stable.gasket
  • linuxKernel.packages.linux_zen.gasket
  • linuxPackages_5_10_hardened.gasket
  • linuxPackages_5_4_hardened.gasket
  • linuxPackages_lqx.gasket
  • linuxPackages_xanmod.gasket
  • linuxPackages_xanmod_latest.gasket
  • linuxPackages_xanmod_stable.gasket
  • linuxPackages_zen.gasket
✅ 14 packages built:
  • linuxKernel.packages.linux_5_15.gasket
  • linuxPackages_5_15_hardened.gasket (linuxKernel.packages.linux_5_15_hardened.gasket)
  • linuxKernel.packages.linux_6_1.gasket
  • linuxPackages.gasket (linuxKernel.packages.linux_6_12.gasket)
  • linuxPackages_hardened.gasket (linuxPackages_6_12_hardened.gasket)
  • linuxKernel.packages.linux_6_13.gasket
  • linuxPackages_6_13_hardened.gasket (linuxKernel.packages.linux_6_13_hardened.gasket)
  • linuxPackages_latest.gasket (linuxKernel.packages.linux_6_14.gasket)
  • linuxPackages_6_1_hardened.gasket (linuxKernel.packages.linux_6_1_hardened.gasket)
  • linuxKernel.packages.linux_6_6.gasket
  • linuxPackages_6_6_hardened.gasket (linuxKernel.packages.linux_6_6_hardened.gasket)
  • linuxPackages_ham.gasket (linuxKernel.packages.linux_ham.gasket)
  • linuxPackages_latest-libre.gasket (linuxKernel.packages.linux_latest_libre.gasket)
  • linuxPackages-libre.gasket (linuxKernel.packages.linux_libre.gasket)

@wegank wegank added 12.approvals: 2 This PR was reviewed and approved by two persons. and removed 12.approvals: 1 This PR was reviewed and approved by one person. labels May 25, 2025
@kylehendricks
kylehendricks requested a review from drupol June 20, 2025 01:00
@drupol
drupol merged commit f2e68be into NixOS:master Jun 20, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

10.rebuild-darwin: 0 This PR does not cause any packages to rebuild on Darwin. 10.rebuild-linux: 11-100 This PR causes between 11 and 100 packages to rebuild on Linux. 12.approvals: 2 This PR was reviewed and approved by two persons. 12.approved-by: package-maintainer This PR was reviewed and approved by a maintainer listed in any of the changed packages. 12.first-time contribution This PR is the author's first one; please be gentle!

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants