Skip to content

fix: actually regenerate icons - #11

Merged
IgnaceMaes merged 2 commits into
mainfrom
actually-regenerate
Sep 14, 2025
Merged

fix: actually regenerate icons#11
IgnaceMaes merged 2 commits into
mainfrom
actually-regenerate

Conversation

@IgnaceMaes

@IgnaceMaes IgnaceMaes commented Sep 14, 2025

Copy link
Copy Markdown
Owner

So turns out it does not run in CI!

@changeset-bot

changeset-bot Bot commented Sep 14, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1b8087f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
ember-phosphor-icons Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@IgnaceMaes
IgnaceMaes merged commit 7ae65dd into main Sep 14, 2025
10 checks passed
@github-actions github-actions Bot mentioned this pull request Sep 14, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant