Skip to content

Update dependency google-cloud-pubsub to v0.45.0 - #29

Open
renovate-bot wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
renovate-bot:renovate/google-cloud-pubsub-0.x
Open

renovate-bot wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
renovate-bot:renovate/google-cloud-pubsub-0.x

Conversation

@renovate-bot

@renovate-bot renovate-bot commented Jul 29, 2019

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
google-cloud-pubsub (source) ==0.42.1==0.45.0 age confidence

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@helmsman-bot

Copy link
Copy Markdown

Can one of the admins verify this patch?

@google-oss-robot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: renovate-bot
To fully approve this pull request, please assign additional approvers.
We suggest the following additional approver: krzyzacy

If they are not already assigned, you can assign the PR to them by writing /assign @krzyzacy in a comment when ready.

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@google-oss-robot google-oss-robot added gke-helmsman PSO helmsman project size/XS labels Jul 29, 2019
@renovate-bot
renovate-bot force-pushed the renovate/google-cloud-pubsub-0.x branch from 89e6ed0 to 5e13b87 Compare July 31, 2019 18:48
@renovate-bot renovate-bot changed the title Update dependency google-cloud-pubsub to v0.44.0 Update dependency google-cloud-pubsub to v0.45.0 Jul 31, 2019
@chrislovecnm

Copy link
Copy Markdown
Contributor

test this please

@renovate-bot
renovate-bot force-pushed the renovate/google-cloud-pubsub-0.x branch from 5e13b87 to 62a9fae Compare August 10, 2025 16:04
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants