Skip to content

Changed project name to project id to be consistent with gcloud comma… - #10

Open
piyush123 wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
piyush123:master
Open

piyush123 wants to merge 1 commit into
GoogleCloudPlatform:masterfrom
piyush123:master

Conversation

@piyush123

Copy link
Copy Markdown
Member

…nd reference

@googlebot

Copy link
Copy Markdown

Thanks for your pull request. It looks like this may be your first contribution to a Google open source project (if not, look below for help). Before we can look at your pull request, you'll need to sign a Contributor License Agreement (CLA).

📝 Please visit https://cla.developers.google.com/ to sign.

Once you've signed (or fixed any issues), please reply here (e.g. I signed it!) and we'll verify it.


What to do if you already signed the CLA

Individual signers
Corporate signers

@google-oss-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
To fully approve this pull request, please assign additional approvers.
We suggest the following additional approver: erkolson

If they are not already assigned, you can assign the PR to them by writing /assign @erkolson in a comment when ready.

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@chrislovecnm

Copy link
Copy Markdown
Contributor

@piyush123 thanks for the Update. Do you mind signing the CLA?

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants