Skip to content

ci: Increase max_locks_per_transaction for Python tests - #6569

Merged
khvn26 merged 5 commits into
ci/private-packages-test-improvementsfrom
ci/max-locks-per-transaction
Jan 21, 2026
Merged

khvn26 merged 5 commits into
ci/private-packages-test-improvementsfrom
ci/max-locks-per-transaction

Conversation

@khvn26

@khvn26 khvn26 commented Jan 21, 2026

Copy link
Copy Markdown
Member

Thanks for submitting a PR! Please check the boxes below:

  • I have read the Contributing Guide.
  • I have added information to docs/ if required so people know about the feature.
  • I have filled in the "Changes" section below.
  • I have filled in the "How did you test this code" section below.

Changes

In this PR, we attempt to fix the django.db.utils.OperationalError: out of shared memory error in CI by increasing default max_locks_per_transaction for the test databases. As a side effect, we move from Postgres services defined in Github Actions to a more centralised Compose-based approach.

How did you test this code?

This is a CI change.

@vercel

vercel Bot commented Jan 21, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

3 Skipped Deployments
Project Deployment Review Updated (UTC)
docs Ignored Ignored Preview Jan 21, 2026 10:42am
flagsmith-frontend-preview Ignored Ignored Preview Jan 21, 2026 10:42am
flagsmith-frontend-staging Ignored Ignored Preview Jan 21, 2026 10:42am

Request Review

@github-actions github-actions Bot added api Issue related to the REST API ci-cd Build, test and deployment related and removed ci-cd Build, test and deployment related labels Jan 21, 2026
@khvn26
khvn26 changed the base branch from main to ci/private-packages-test-improvements January 21, 2026 10:23
@khvn26
khvn26 marked this pull request as ready for review January 21, 2026 10:23
@khvn26
khvn26 requested review from a team as code owners January 21, 2026 10:23
@khvn26
khvn26 requested review from gagantrivedi and removed request for a team January 21, 2026 10:23
@github-actions github-actions Bot added ci-cd Build, test and deployment related and removed ci-cd Build, test and deployment related labels Jan 21, 2026
@github-actions

github-actions Bot commented Jan 21, 2026

Copy link
Copy Markdown
Contributor

Docker builds report

Image Build Status Security report
ghcr.io/flagsmith/flagsmith-api-test:pr-6569 Finished ✅ Skipped
ghcr.io/flagsmith/flagsmith-e2e:pr-6569 Finished ✅ Skipped
ghcr.io/flagsmith/flagsmith-frontend:pr-6569 Finished ✅ Results
ghcr.io/flagsmith/flagsmith-api:pr-6569 Finished ✅ Results
ghcr.io/flagsmith/flagsmith:pr-6569 Finished ✅ Results
ghcr.io/flagsmith/flagsmith-private-cloud:pr-6569 Finished ✅ Results

@codecov

codecov Bot commented Jan 21, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 98.15%. Comparing base (2cb22c4) to head (c0a9844).
⚠️ Report is 1 commits behind head on ci/private-packages-test-improvements.

Additional details and impacted files
@@                          Coverage Diff                           @@
##           ci/private-packages-test-improvements    #6569   +/-   ##
======================================================================
  Coverage                                  98.15%   98.15%           
======================================================================
  Files                                       1293     1293           
  Lines                                      46610    46610           
======================================================================
  Hits                                       45751    45751           
  Misses                                       859      859           

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@khvn26
khvn26 force-pushed the ci/max-locks-per-transaction branch from 4f8669a to ef36df9 Compare January 21, 2026 10:24
@github-actions github-actions Bot added ci-cd Build, test and deployment related and removed ci-cd Build, test and deployment related labels Jan 21, 2026
@github-actions github-actions Bot added ci-cd Build, test and deployment related and removed ci-cd Build, test and deployment related labels Jan 21, 2026
@github-actions github-actions Bot added ci-cd Build, test and deployment related and removed ci-cd Build, test and deployment related labels Jan 21, 2026
@khvn26
khvn26 merged commit 6ac214c into ci/private-packages-test-improvements Jan 21, 2026
19 checks passed
@khvn26
khvn26 deleted the ci/max-locks-per-transaction branch January 21, 2026 10:44
khvn26 added a commit that referenced this pull request Jan 21, 2026
ci: Increase max_locks_per_transaction for Python tests (#6569)
--
Co-authored-by: Kim Gustyr <kim.gustyr@flagsmith.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

api Issue related to the REST API ci-cd Build, test and deployment related

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants