Skip to content

feat: clean-site-packages-before-mv - #6568

Closed
Zaimwa9 wants to merge 1 commit into
mainfrom
fix/clean-up-site-packages-for-ci
Closed

Zaimwa9 wants to merge 1 commit into
mainfrom
fix/clean-up-site-packages-for-ci

Conversation

@Zaimwa9

@Zaimwa9 Zaimwa9 commented Jan 21, 2026

Copy link
Copy Markdown
Contributor

Thanks for submitting a PR! Please check the boxes below:

  • I have read the Contributing Guide.
  • I have added information to docs/ if required so people know about the feature.
  • I have filled in the "Changes" section below.
  • I have filled in the "How did you test this code" section below.

Changes

API tests are failing due to poetry cache during CI cf here

  • Cleans-up the site packages before trying to install them

How did you test this code?

  • CI

@Zaimwa9
Zaimwa9 requested a review from a team as a code owner January 21, 2026 09:29
@Zaimwa9
Zaimwa9 requested review from emyller and removed request for a team January 21, 2026 09:29
@vercel

vercel Bot commented Jan 21, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

3 Skipped Deployments
Project Deployment Review Updated (UTC)
docs Ignored Ignored Jan 21, 2026 9:29am
flagsmith-frontend-preview Ignored Ignored Jan 21, 2026 9:29am
flagsmith-frontend-staging Ignored Ignored Jan 21, 2026 9:29am

Request Review

@github-actions github-actions Bot added api Issue related to the REST API feature New feature or request labels Jan 21, 2026
@github-actions

github-actions Bot commented Jan 21, 2026

Copy link
Copy Markdown
Contributor

Docker builds report

Image Build Status Security report
ghcr.io/flagsmith/flagsmith-e2e:pr-6568 Finished ✅ Skipped
ghcr.io/flagsmith/flagsmith-frontend:pr-6568 Finished ✅ Results
ghcr.io/flagsmith/flagsmith-api-test:pr-6568 Finished ✅ Skipped
ghcr.io/flagsmith/flagsmith-api:pr-6568 Finished ✅ Results
ghcr.io/flagsmith/flagsmith:pr-6568 Finished ✅ Results
ghcr.io/flagsmith/flagsmith-private-cloud:pr-6568 Finished ✅ Results

@matthewelwell

Copy link
Copy Markdown
Contributor

Duplicate of #6566

@matthewelwell matthewelwell marked this as a duplicate of #6566 Jan 21, 2026
@codecov

codecov Bot commented Jan 21, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 98.15%. Comparing base (217f217) to head (83bb10b).

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #6568   +/-   ##
=======================================
  Coverage   98.15%   98.15%           
=======================================
  Files        1293     1293           
  Lines       46610    46610           
=======================================
  Hits        45751    45751           
  Misses        859      859           

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@emyller
emyller deleted the fix/clean-up-site-packages-for-ci branch January 21, 2026 12:16
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

api Issue related to the REST API feature New feature or request

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants