Skip to content

docs(writechoice): remove openfeature python beta status - #6048

Merged
emyller merged 1 commit into
docs/writechoice-dev-docsfrom
docs/remove-openfeature-python-beta-status
Sep 10, 2025
Merged

docs(writechoice): remove openfeature python beta status#6048
emyller merged 1 commit into
docs/writechoice-dev-docsfrom
docs/remove-openfeature-python-beta-status

Conversation

@matthewelwell

Copy link
Copy Markdown
Contributor

Changes

Replicates the changes made in #6047 to the writechoice dev docs branch (note target branch is docs/writechoice-dev-docs).

How did you test this code?

N/a

@matthewelwell
matthewelwell requested a review from a team as a code owner September 10, 2025 21:55
@matthewelwell
matthewelwell requested review from adamvialpando and removed request for a team September 10, 2025 21:55
@vercel

vercel Bot commented Sep 10, 2025

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
docs Ready Ready Preview Comment Sep 10, 2025 9:55pm
2 Skipped Deployments
Project Deployment Preview Comments Updated (UTC)
flagsmith-frontend-preview Ignored Ignored Sep 10, 2025 9:55pm
flagsmith-frontend-staging Ignored Ignored Sep 10, 2025 9:55pm

@emyller
emyller merged commit 49de383 into docs/writechoice-dev-docs Sep 10, 2025
9 checks passed
@emyller
emyller deleted the docs/remove-openfeature-python-beta-status branch September 10, 2025 22:23
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docs Documentation updates

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants