Skip to content

Adjust sysctl bash template to use build time variables - #15017

Merged
ggbecker merged 1 commit into
ComplianceAsCode:masterfrom
Mab879:fix_up_2026_08_19
Aug 21, 2026
Merged

Adjust sysctl bash template to use build time variables#15017
ggbecker merged 1 commit into
ComplianceAsCode:masterfrom
Mab879:fix_up_2026_08_19

Conversation

@Mab879

@Mab879 Mab879 commented Aug 19, 2026

Copy link
Copy Markdown
Member

Description:

Adjust sysctl bash template to use build time variables

Rationale:

Should fix a few issues with bash substitution.

@Mab879 Mab879 added this to the 0.1.82 milestone Aug 19, 2026
@Mab879 Mab879 added the backported-into-stabilization PRs which were cherry-picked during stabilization process. label Aug 19, 2026
@openshift-ci

openshift-ci Bot commented Aug 19, 2026

Copy link
Copy Markdown

@Mab879: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-aws-openshift-platform-compliance 58f096b link true /test e2e-aws-openshift-platform-compliance
ci/prow/e2e-aws-openshift-node-compliance 58f096b link false /test e2e-aws-openshift-node-compliance

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Comment thread shared/templates/sysctl/bash.template Outdated
#
{{% if sysctl_remediate_drop_in_file == "true" %}}
sed -i "/^$SYSCONFIG_VAR/d" /etc/sysctl.conf
sed -i "/^{{{ SYSCONFIG_VAR }}}/d" /etc/sysctl.conf

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I struggle to find the definition of SYSCONFIG_VAR though...

Should we use something like

Suggested change
sed -i "/^{{{ SYSCONFIG_VAR }}}/d" /etc/sysctl.conf
sed -i "/^{{{ SYSCTLVAR }}}/d" /etc/sysctl.conf

here?

@Mab879
Mab879 force-pushed the fix_up_2026_08_19 branch from 58f096b to 8e92507 Compare August 21, 2026 14:10
@Mab879
Mab879 force-pushed the fix_up_2026_08_19 branch from 8e92507 to cb22069 Compare August 21, 2026 14:23
@ggbecker ggbecker self-assigned this Aug 21, 2026
@ggbecker

Copy link
Copy Markdown
Member

no need to wait for the ansible lint to finish (which seems stuck), it's only a bash related change what the PR is proposing.

@ggbecker
ggbecker merged commit b3a2cca into ComplianceAsCode:master Aug 21, 2026
72 of 75 checks passed
@Mab879
Mab879 deleted the fix_up_2026_08_19 branch August 21, 2026 17:31
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

backported-into-stabilization PRs which were cherry-picked during stabilization process.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants